Skip to content

Upgrade firefox-syncserver to 0.22.1 and expose PostgreSQL support - #507460

Closed
ntninja wants to merge 2 commits into
NixOS:masterfrom
ntninja:patch-ffsync-postgres
Closed

ntninja wants to merge 2 commits into
NixOS:masterfrom
ntninja:patch-ffsync-postgres

Conversation

@ntninja

@ntninja ntninja commented Apr 7, 2026

Copy link
Copy Markdown
Contributor

Fixes #484932

This upgrade the syncstorage-rs package to new 0.22.1 version and splits its build into a MySQL and (new) PostgreSQL variant. The NixOS module is upgraded to allow the usage for PostgreSQL with a simple option change.

Existing NixOS installation will continue to use MySQL, new installations will need to explicitly specify the database backend.

@mweinelt @n0rc

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin (N/A)
    • aarch64-darwin (N/A)
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking. (PostgreSQL support)
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.(PostgreSQL support)
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

ntninja added 2 commits April 7, 2026 04:07
…variant

Split package into two subpackages for MySQL and PostgreSQL each. The previous
`syncstorage-rs` package name points at the MySQL variant for backward
compatiblity.
@ntninja

ntninja commented Apr 7, 2026

Copy link
Copy Markdown
Contributor Author

It’s complaining that I declared the package in ./all-packages.nix to reference the different variants. Some review on whether the chosen approach for declaring the variants is OK to begin with would be appreciated.

@mweinelt

mweinelt commented Apr 8, 2026

Copy link
Copy Markdown
Member

Duplicate of #501943

@mweinelt mweinelt marked this as a duplicate of #501943 Apr 8, 2026
@ntninja

ntninja commented Apr 8, 2026

Copy link
Copy Markdown
Contributor Author

Did a review of their version and submitted comments there. I may have missed a thing or two, but most things that are better in mine than theirs should now also be in their review comments:
#501943 (review)

@nixpkgs-ci nixpkgs-ci Bot added the 2.status: merge conflict This PR has merge conflicts with the target branch label May 21, 2026
@kira-bruneau

Copy link
Copy Markdown
Contributor

Replaced with #446862 & #501943

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge conflict This PR has merge conflicts with the target branch

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants